Skip to content

[test_operator] Remove HorizonTest credentials WA - #4157

Draft
kstrenkova wants to merge 1 commit into
openstack-k8s-operators:mainfrom
kstrenkova:remove-horizontest-credentials-WA
Draft

[test_operator] Remove HorizonTest credentials WA#4157
kstrenkova wants to merge 1 commit into
openstack-k8s-operators:mainfrom
kstrenkova:remove-horizontest-credentials-WA

Conversation

@kstrenkova

Copy link
Copy Markdown
Contributor

There is a HorizonTest WA for credentials due to a change from static to dynamic passwords. This PR aims to remove this WA since the actual fix has reached downstream and we do not have to use WA anymore.

There is a HorizonTest WA for credentials due to a change from
static to dynamic passwords. This PR aims to remove this WA since
the actual fix has reached downstream and we do not have to use
WA anymore.

Signed-off-by: Katarina Strenkova <kstrenko@redhat.com>
@openshift-ci

openshift-ci Bot commented Sep 3, 2026

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@openshift-ci

openshift-ci Bot commented Sep 3, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign kstrenkova for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kstrenkova

kstrenkova commented Sep 3, 2026

Copy link
Copy Markdown
Contributor Author

I have created this PR to test whether we can actually remove the WA. Once I get a passing test job, I will mark it here. Until then this is a work in progress.

Edit: After running the job, it seems the test-operator change is already downstream, but the tcib change has yet to propagate. I am keeping this ticket as a draft and will try to run it again next week.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/a1343168a3e54586a55d94550f4a309c

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 37m 50s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 47m 17s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 2h 00m 48s
✔️ cifmw-crc-podified-edpm-baremetal-minor-update SUCCESS in 2h 14m 00s
✔️ cifmw-multinode-tempest SUCCESS in 2h 10m 55s
✔️ cifmw-pod-zuul-files SUCCESS in 7m 58s
✔️ openstack-k8s-operators-content-provider-bootc SUCCESS in 2h 24m 50s
cifmw-crc-podified-edpm-baremetal-bootc FAILURE in 1h 44m 25s
✔️ noop SUCCESS in 0s
✔️ cifmw-pod-ansible-test SUCCESS in 12m 36s
✔️ cifmw-pod-pre-commit SUCCESS in 12m 25s
✔️ cifmw-molecule-test_operator SUCCESS in 4m 34s

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant