Skip to content

Fix typos and punctuation in STYLE-GUIDE.md - #502

Open
YasinDehfuli wants to merge 1 commit into
reactjs:mainfrom
YasinDehfuli:patch-1
Open

YasinDehfuli wants to merge 1 commit into
reactjs:mainfrom
YasinDehfuli:patch-1

Conversation

@YasinDehfuli

Copy link
Copy Markdown

Corrected typographical errors and improved punctuation for clarity.

Corrected typographical errors and improved punctuation for clarity.
@vercel

vercel Bot commented Oct 1, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Preview Comments Updated (UTC)
fa-legacy-reactjs-org Ignored Ignored Preview Oct 1, 2025 1:05pm

This branch was successfully deployed

1 active deployment
Preview – fa-react-dev — bd491352 Deployed Oct 1, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant